Skip to content

Add Russian translation for components-form-page - #1035

Open
dimentoree wants to merge 1 commit into
reactjs:mainfrom
dimentoree:translate-components-form-page
Open

dimentoree wants to merge 1 commit into
reactjs:mainfrom
dimentoree:translate-components-form-page

Conversation

@dimentoree

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

vercel Bot commented Aug 9, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
ru-legacy-reactjs-org ⬜️ Ignored (Inspect) Visit Preview Aug 9, 2025 10:24am

This branch was successfully deployed

1 active deployment
Preview – ru-react-dev — b1de4126 Deployed Aug 9, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant