Skip to content

docs: translate preinit - #765

Merged
resir014 merged 2 commits into
reactjs:mainfrom
bagoesrex:docs/translate-preinit.md
Jun 16, 2025
Merged

resir014 merged 2 commits into
reactjs:mainfrom
bagoesrex:docs/translate-preinit.md

Conversation

@bagoesrex

@bagoesrex bagoesrex commented Jun 15, 2025 •

Copy link
Copy Markdown
Contributor

Closes #693

Description

Translate the preinit page.
Page URL: https://id.react.dev/reference/react-dom/preinit

@vercel

vercel Bot commented Jun 15, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
id-react-dev 🔄 Building (Inspect) Visit Preview 💬 Add feedback Jun 16, 2025 11:51am
1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
id-legacy-reactjs-org ⬜️ Ignored (Inspect) Visit Preview Jun 16, 2025 11:51am

@resir014 resir014 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for your help! @bagoesrex

@github-actions

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@resir014
resir014 merged commit 5902392 into reactjs:main Jun 16, 2025

This branch was successfully deployed

1 active deployment
Preview – id-react-dev — 58bb4a62 Deployed Jun 16, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

preinit

2 participants