fix(clerk-js): Trigger re-verification when changing primary email - #5162
Merged
octoper merged 2 commits intoFeb 14, 2025
Conversation
🦋 Changeset detectedLatest commit: 356246f The changes in this PR will be included in the next version bump. This PR includes changesets to release 3 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎
|
nikospapcom
approved these changes
Feb 14, 2025
panteliselef
approved these changes
Feb 14, 2025
octoper
force-pushed
the
vaggelis/user-1200-apply-reverification-when-changing-primary-identification
branch
from
February 14, 2025 15:37
2830c50 to
356246f
Compare
octoper
enabled auto-merge (squash)
February 14, 2025 15:38
octoper
deleted the
vaggelis/user-1200-apply-reverification-when-changing-primary-identification
branch
February 14, 2025 15:49
3 of 9 tasks
This branch was successfully deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description
This PR introduces re-verification when trying to change primary email address
Checklist
pnpm testruns as expected.pnpm buildruns as expected.Type of change