Skip to content

feat(clerk-js): Retheme pages to cards - #2349

Merged
SokratisVidros merged 6 commits into
mainfrom
retheme_pages_to_cards
Dec 15, 2023
Merged

SokratisVidros merged 6 commits into
mainfrom
retheme_pages_to_cards

Conversation

@desiprisg

@desiprisg desiprisg commented Dec 14, 2023 •

Copy link
Copy Markdown
Contributor

Description

This PR mainly rethemes the UserProfile and all the subpages to cards.

Checklist

  • npm test runs as expected.
  • npm run build runs as expected.
  • (If applicable) JSDoc comments have been added or updated for any package exports
  • (If applicable) Documentation has been updated

Type of change

  • 🐛 Bug fix
  • 🌟 New feature
  • 🔨 Breaking change
  • 📖 Refactoring / dependency upgrade / documentation
  • other:

Packages affected

  • @clerk/backend
  • @clerk/chrome-extension
  • @clerk/clerk-js
  • @clerk/clerk-expo
  • @clerk/fastify
  • gatsby-plugin-clerk
  • @clerk/localizations
  • @clerk/nextjs
  • @clerk/clerk-react
  • @clerk/remix
  • @clerk/clerk-sdk-node
  • @clerk/shared
  • @clerk/themes
  • @clerk/types
  • build/tooling/chore

@changeset-bot

changeset-bot Bot commented Dec 14, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 525ca2b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 0 packages

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@desiprisg desiprisg changed the title feat(clerk-js): Implement UsernameForm in UsernameSection feat(clerk-js): Retheme pages to cards Dec 14, 2023
@desiprisg
desiprisg force-pushed the retheme_pages_to_cards branch 3 times, most recently from 0d26721 to feccba8 Compare December 14, 2023 19:57
@desiprisg
desiprisg force-pushed the retheme_pages_to_cards branch 3 times, most recently from faf3215 to ff7b1c6 Compare December 15, 2023 10:28
@desiprisg
desiprisg marked this pull request as ready for review December 15, 2023 11:19
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants